This repository was archived by the owner on Apr 3, 2020. It is now read-only.
Conversation
yunxliu
force-pushed
the
deviceorientation/cts-1742
branch
from
May 27, 2016 05:45
f512d10 to
6ddb12d
Compare
Impacted tests(approved): new 6, update 0, delete 0 Unit test platform: Crosswalk Project for Android 20.50.531.0 Unit test result summary: pass 6, fail 0, block 0 BUG=https://crosswalk-project.org/jira/browse/CTS-1742
yunxliu
force-pushed
the
deviceorientation/cts-1742
branch
from
May 27, 2016 07:19
6ddb12d to
c513cd9
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to subscribe to this conversation on GitHub.
Already have an account?
Sign in.
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Impacted tests(approved): new 5, update 0, delete 0
Unit test platform: Crosswalk Project for Android 20.50.531.0
Unit test result summary: pass 5, fail 0, block 0
BUG=https://crosswalk-project.org/jira/browse/CTS-1742