Skip to content

Add map library with merge functionality - #1068

Merged
k8s-ci-robot merged 2 commits into
kubernetes-sigs:mainfrom
a-hilaly:gfrey/add_maps_merge
Feb 24, 2026
Merged

Add map library with merge functionality#1068
k8s-ci-robot merged 2 commits into
kubernetes-sigs:mainfrom
a-hilaly:gfrey/add_maps_merge

Conversation

@a-hilaly

Copy link
Copy Markdown
Member

Allows to merge two maps. Maps must have identical value types. Keys of
the second map take precedence.

Brought from https://github.com/google/cel-go/pull/1239

Allows to merge two maps. Maps must have identical value types. Keys of
the second map take precedence.
@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Feb 24, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: a-hilaly

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added approved Indicates a PR has been approved by an approver from all required OWNERS files. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Feb 24, 2026
@a-hilaly

Copy link
Copy Markdown
Member Author

/cc @gfrey

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@a-hilaly: GitHub didn't allow me to request PR reviews from the following users: gfrey.

Note that only kubernetes-sigs members and repo collaborators can review this PR, and authors cannot review their own PRs.

Details

In response to this:

/cc @gfrey

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@jakobmoellerdev jakobmoellerdev left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 24, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 1b7cee8 into kubernetes-sigs:main Feb 24, 2026
7 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants