Skip to content

feat: merge next-sbcom changes - #3060

Open
TitanKuzmich wants to merge 1 commit into
devfrom
PLASMA-merge-sbcom
Open

feat: merge next-sbcom changes#3060
TitanKuzmich wants to merge 1 commit into
devfrom
PLASMA-merge-sbcom

Conversation

@TitanKuzmich

@TitanKuzmich TitanKuzmich commented Aug 18, 2026

Copy link
Copy Markdown
Collaborator

What/why changed

merge next-sbcom into dev

  • после публикации next в этот пр влить актуальную ветку
  • провести регресс ветки
  • влить через merge commit в dev ветку

@TitanKuzmich TitanKuzmich added the changelog-skip Exclude pull request changes from changelog label Aug 18, 2026
@coderabbitai

coderabbitai Bot commented Aug 18, 2026

Copy link
Copy Markdown

Important

Review skipped

Too many files!

This PR contains 345 files, which is 245 over the limit of 100.

To get a review, reduce the PR to 100 files or fewer by splitting it into smaller PRs or changing its base branch.

Upgrade to a paid plan to raise the limit.

Usage-priced reviews support at most 300 files.

⚙️ Run configuration

Configuration used: defaults

Review profile: CHILL

Plan: Pro Plus

Run ID: b9c9b7b7-8c10-4e07-a158-a355818119a6

📥 Commits

Reviewing files that changed from the base of the PR and between 7e7f2b2 and 3651d74.

⛔ Files ignored due to path filters (100)
  • cypress/snapshots/homeds/chromium/TextArea/TextArea -- hintPlacement=right.snap.png is excluded by !**/*.png
  • cypress/snapshots/homeds/chromium/TextArea/TextArea -- hintSize=s, hintPlacement=bottom, hintWidth=100px.snap.png is excluded by !**/*.png
  • cypress/snapshots/homeds/chromium/TextArea/TextArea -- hintTrigger=hover, hintSize=m, hintPlacement=top, hintWidth=1px, hintHasArrow.snap.png is excluded by !**/*.png
  • cypress/snapshots/homeds/webkit/TextArea/TextArea -- hintPlacement=right.snap.png is excluded by !**/*.png
  • cypress/snapshots/homeds/webkit/TextArea/TextArea -- hintSize=s, hintPlacement=bottom, hintWidth=100px.snap.png is excluded by !**/*.png
  • cypress/snapshots/homeds/webkit/TextArea/TextArea -- hintTrigger=click, hintHasArrow=false.snap.png is excluded by !**/*.png
  • cypress/snapshots/homeds/webkit/TextArea/TextArea -- hintTrigger=hover, hintSize=m, hintPlacement=top, hintWidth=1px, hintHasArrow.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/Calendar -- case very future date.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/Calendar -- default double calendar.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/Calendar -- default.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/Calendar -- locale en.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/Calendar -- min and max double.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/Calendar -- min and max.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/Calendar -- size=s.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/Calendar -- stretch double.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/Calendar -- stretch single.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/Calendar keyboard navigation -- navigate with Down, Left, Up, Right arrows.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/Calendar keyboard navigation -- set value with EnterandSpace.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/CalendarRange -- default double calendar.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/CalendarRange -- range in progress with disabled.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol-next/chromium/Calendar/CalendarRange -- range in progress.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/Calendar -- case very future date.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/Calendar -- default double calendar.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/Calendar -- default.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/Calendar -- locale en.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/Calendar -- min and max double.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/Calendar -- min and max.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/Calendar -- size=s.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/Calendar -- stretch double.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/Calendar -- stretch single.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/Calendar keyboard navigation -- navigate with Down, Left, Up, Right arrows.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/Calendar keyboard navigation -- set value with EnterandSpace.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/CalendarRange -- default double calendar.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/CalendarRange -- range in progress with disabled.snap.png is excluded by !**/*.png
  • cypress/snapshots/insol/chromium/Calendar/CalendarRange -- range in progress.snap.png is excluded by !**/*.png
  • cypress/snapshots/platform-ai/chromium/TextArea/TextArea -- hintPlacement=right.snap.png is excluded by !**/*.png
  • cypress/snapshots/platform-ai/chromium/TextArea/TextArea -- hintSize=s, hintPlacement=bottom, hintWidth=100px.snap.png is excluded by !**/*.png
  • cypress/snapshots/platform-ai/chromium/TextArea/TextArea -- hintTrigger=hover, hintSize=m, hintPlacement=top, hintWidth=1px, hintHasArrow.snap.png is excluded by !**/*.png
  • cypress/snapshots/platform-ai/webkit/TextArea/TextArea -- disabled.snap.png is excluded by !**/*.png
  • cypress/snapshots/platform-ai/webkit/TextArea/TextArea -- hintPlacement=right.snap.png is excluded by !**/*.png
  • cypress/snapshots/platform-ai/webkit/TextArea/TextArea -- hintSize=s, hintPlacement=bottom, hintWidth=100px.snap.png is excluded by !**/*.png
  • cypress/snapshots/platform-ai/webkit/TextArea/TextArea -- hintTrigger=click, hintHasArrow=false.snap.png is excluded by !**/*.png
  • cypress/snapshots/platform-ai/webkit/TextArea/TextArea -- hintTrigger=hover, hintSize=m, hintPlacement=top, hintWidth=1px, hintHasArrow.snap.png is excluded by !**/*.png
  • cypress/snapshots/scan/chromium/TextArea/TextArea -- view=default, size=m appearance=clear,hasDivider=true.snap.png is excluded by !**/*.png
  • cypress/snapshots/scan/chromium/TextArea/TextArea -- view=negative appearance=clear,hasDivider=true.snap.png is excluded by !**/*.png
  • cypress/snapshots/scan/chromium/TextArea/TextArea -- view=positive, size=s appearance=clear,hasDivider=true.snap.png is excluded by !**/*.png
  • cypress/snapshots/scan/chromium/TextArea/TextArea -- view=warning appearance=clear,hasDivider=true.snap.png is excluded by !**/*.png
  • cypress/snapshots/scan/webkit/TextArea/TextArea -- view=default, size=m appearance=clear,hasDivider=true.snap.png is excluded by !**/*.png
  • cypress/snapshots/scan/webkit/TextArea/TextArea -- view=negative appearance=clear,hasDivider=true.snap.png is excluded by !**/*.png
  • cypress/snapshots/scan/webkit/TextArea/TextArea -- view=positive, size=s appearance=clear,hasDivider=true.snap.png is excluded by !**/*.png
  • cypress/snapshots/scan/webkit/TextArea/TextArea -- view=warning appearance=clear,hasDivider=true.snap.png is excluded by !**/*.png
  • package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-asdk/package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-b2c/package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-core/package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-giga/package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-homeds/package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-hope/package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-icons/package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-new-hope/package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-tokens-b2b/package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-tokens-b2c/package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-tokens-web/package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-tokens/package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-typo/package-lock.json is excluded by !**/package-lock.json
  • packages/plasma-web/package-lock.json is excluded by !**/package-lock.json
  • packages/sdds-bizcom/package-lock.json is excluded by !**/package-lock.json
  • packages/sdds-cs/package-lock.json is excluded by !**/package-lock.json
  • packages/sdds-dfa/package-lock.json is excluded by !**/package-lock.json
  • packages/sdds-finai/package-lock.json is excluded by !**/package-lock.json
  • packages/sdds-insol/package-lock.json is excluded by !**/package-lock.json
  • packages/sdds-netology/package-lock.json is excluded by !**/package-lock.json
  • packages/sdds-os/package-lock.json is excluded by !**/package-lock.json
  • packages/sdds-platform-ai/package-lock.json is excluded by !**/package-lock.json
  • packages/sdds-sbcom/package-lock.json is excluded by !**/package-lock.json
  • packages/sdds-scan/package-lock.json is excluded by !**/package-lock.json
  • packages/sdds-serv/package-lock.json is excluded by !**/package-lock.json
  • packages/themes/core-themes/package-lock.json is excluded by !**/package-lock.json
  • packages/themes/plasma-themes/package-lock.json is excluded by !**/package-lock.json
  • packages/themes/sdds-themes/package-lock.json is excluded by !**/package-lock.json
  • utils/api-tests/package-lock.json is excluded by !**/package-lock.json
  • utils/plasma-cy-utils/package-lock.json is excluded by !**/package-lock.json
  • utils/plasma-docs-ui/package-lock.json is excluded by !**/package-lock.json
  • utils/plasma-sb-utils/package-lock.json is excluded by !**/package-lock.json
  • utils/plasma-tokens-utils/package-lock.json is excluded by !**/package-lock.json
  • website/plasma-b2c-docs/package-lock.json is excluded by !**/package-lock.json
  • website/plasma-giga-docs/package-lock.json is excluded by !**/package-lock.json
  • website/plasma-homeds-docs/package-lock.json is excluded by !**/package-lock.json
  • website/plasma-web-docs/package-lock.json is excluded by !**/package-lock.json
  • website/plasma-website/package-lock.json is excluded by !**/package-lock.json
  • website/sdds-bizcom-docs/package-lock.json is excluded by !**/package-lock.json
  • website/sdds-cs-docs/package-lock.json is excluded by !**/package-lock.json
  • website/sdds-dfa-docs/package-lock.json is excluded by !**/package-lock.json
  • website/sdds-finai-docs/package-lock.json is excluded by !**/package-lock.json
  • website/sdds-insol-docs/package-lock.json is excluded by !**/package-lock.json
  • website/sdds-netology-docs/package-lock.json is excluded by !**/package-lock.json
  • website/sdds-platform-ai-docs/package-lock.json is excluded by !**/package-lock.json
  • website/sdds-sbcom-docs/package-lock.json is excluded by !**/package-lock.json
  • website/sdds-scan-docs/package-lock.json is excluded by !**/package-lock.json
  • website/sdds-serv-docs/package-lock.json is excluded by !**/package-lock.json
📒 Files selected for processing (345)
  • .eslintrc.js
  • packages/plasma-asdk/package.json
  • packages/plasma-b2c/package.json
  • packages/plasma-b2c/src/components/Autocomplete/Autocomplete.config.ts
  • packages/plasma-b2c/src/components/Calendar/Calendar.config.ts
  • packages/plasma-b2c/src/components/Calendar/CalendarDouble.config.ts
  • packages/plasma-b2c/src/components/Combobox/Combobox.config.ts
  • packages/plasma-b2c/src/components/DatePicker/DatePicker.config.ts
  • packages/plasma-b2c/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/plasma-b2c/src/components/Select/Select.config.ts
  • packages/plasma-b2c/src/components/TextField/TextField.clear.config.ts
  • packages/plasma-b2c/src/components/TextField/TextField.config.ts
  • packages/plasma-b2c/src/components/TextField/TextField.stories.tsx
  • packages/plasma-b2c/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/plasma-colors/package.json
  • packages/plasma-core/package.json
  • packages/plasma-giga/package.json
  • packages/plasma-giga/src/components/Autocomplete/Autocomplete.config.ts
  • packages/plasma-giga/src/components/Calendar/Calendar.config.ts
  • packages/plasma-giga/src/components/Calendar/CalendarDouble.config.ts
  • packages/plasma-giga/src/components/Combobox/Combobox.config.ts
  • packages/plasma-giga/src/components/DatePicker/DatePicker.config.ts
  • packages/plasma-giga/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/plasma-giga/src/components/Select/Select.config.ts
  • packages/plasma-giga/src/components/TextField/TextField.clear.config.ts
  • packages/plasma-giga/src/components/TextField/TextField.config.ts
  • packages/plasma-giga/src/components/TextField/TextField.stories.tsx
  • packages/plasma-giga/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/plasma-homeds/package.json
  • packages/plasma-homeds/src/components/Autocomplete/Autocomplete.config.ts
  • packages/plasma-homeds/src/components/Calendar/Calendar.config.ts
  • packages/plasma-homeds/src/components/Calendar/CalendarDouble.config.ts
  • packages/plasma-homeds/src/components/Combobox/Combobox.config.ts
  • packages/plasma-homeds/src/components/DatePicker/DatePicker.config.ts
  • packages/plasma-homeds/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/plasma-homeds/src/components/Select/Select.config.ts
  • packages/plasma-homeds/src/components/TextField/TextField.config.ts
  • packages/plasma-homeds/src/components/TextField/TextField.stories.tsx
  • packages/plasma-homeds/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/plasma-hope/package.json
  • packages/plasma-icons/package.json
  • packages/plasma-new-hope/package.json
  • packages/plasma-new-hope/src/components/Autocomplete/Autocomplete.tokens.ts
  • packages/plasma-new-hope/src/components/Autocomplete/ui/TextField/TextField.styles.ts
  • packages/plasma-new-hope/src/components/Avatar/Avatar.types.ts
  • packages/plasma-new-hope/src/components/Button/Button.tokens.ts
  • packages/plasma-new-hope/src/components/Button/variations/_disabled/base.ts
  • packages/plasma-new-hope/src/components/Calendar/Calendar.tokens.ts
  • packages/plasma-new-hope/src/components/Calendar/CalendarBase/CalendarBase.tsx
  • packages/plasma-new-hope/src/components/Calendar/CalendarDouble/CalendarDouble.tsx
  • packages/plasma-new-hope/src/components/Calendar/mixins/selected.ts
  • packages/plasma-new-hope/src/components/Calendar/ui/CalendarDays/CalendarDays.styles.ts
  • packages/plasma-new-hope/src/components/Calendar/ui/CalendarHeader/CalendarHeader.tsx
  • packages/plasma-new-hope/src/components/Calendar/ui/CalendarHeader/CalendarHeader.types.ts
  • packages/plasma-new-hope/src/components/Calendar/ui/CalendarMonths/CalendarMonths.styles.ts
  • packages/plasma-new-hope/src/components/Calendar/ui/CalendarQuarters/CalendarQuarters.styles.ts
  • packages/plasma-new-hope/src/components/Calendar/ui/CalendarYears/CalendarYears.styles.ts
  • packages/plasma-new-hope/src/components/Calendar/ui/DateStructureItem/DateStructureItem.styles.ts
  • packages/plasma-new-hope/src/components/Calendar/ui/DateStructureItem/DateStructureItem.tsx
  • packages/plasma-new-hope/src/components/Calendar/utils/calendarRangeHelper.ts
  • packages/plasma-new-hope/src/components/Checkbox/Checkbox.tokens.ts
  • packages/plasma-new-hope/src/components/Checkbox/variations/_disabled/base.ts
  • packages/plasma-new-hope/src/components/Chip/Chip.tokens.ts
  • packages/plasma-new-hope/src/components/Chip/variations/_disabled/base.ts
  • packages/plasma-new-hope/src/components/Combobox/Combobox.tokens.ts
  • packages/plasma-new-hope/src/components/Combobox/ui/Target/Target.styles.ts
  • packages/plasma-new-hope/src/components/DatePicker/DatePicker.tokens.ts
  • packages/plasma-new-hope/src/components/DatePicker/DatePickerBase.styles.ts
  • packages/plasma-new-hope/src/components/DateTimePicker/DateTimePicker.tokens.ts
  • packages/plasma-new-hope/src/components/DateTimePicker/ui/CalendarGrid/CalendarGrid.styles.ts
  • packages/plasma-new-hope/src/components/DateTimePicker/ui/Popover/Popover.styles.ts
  • packages/plasma-new-hope/src/components/Divider/Divider.template-docs.mdx
  • packages/plasma-new-hope/src/components/Divider/Divider.tsx
  • packages/plasma-new-hope/src/components/Divider/Divider.types.ts
  • packages/plasma-new-hope/src/components/IconButton/IconButton.styles.ts
  • packages/plasma-new-hope/src/components/IconButton/IconButton.tokens.ts
  • packages/plasma-new-hope/src/components/Loader/Loader.tsx
  • packages/plasma-new-hope/src/components/Radiobox/Radiobox.tokens.ts
  • packages/plasma-new-hope/src/components/Radiobox/variations/_disabled/base.ts
  • packages/plasma-new-hope/src/components/Select/Select.tokens.ts
  • packages/plasma-new-hope/src/components/Select/ui/Target/ui/Textfield/Textfield.styles.tsx
  • packages/plasma-new-hope/src/components/Spinner/Spinner.styles.ts
  • packages/plasma-new-hope/src/components/Spinner/Spinner.tsx
  • packages/plasma-new-hope/src/components/Switch/Switch.tokens.ts
  • packages/plasma-new-hope/src/components/Switch/_disabled/base.ts
  • packages/plasma-new-hope/src/components/TextArea/TextArea.styles.ts
  • packages/plasma-new-hope/src/components/TextArea/TextArea.tokens.ts
  • packages/plasma-new-hope/src/components/TextField/TextField.styles.ts
  • packages/plasma-new-hope/src/components/TextField/TextField.tokens.ts
  • packages/plasma-new-hope/src/components/TextField/variations/_size/base.ts
  • packages/plasma-new-hope/src/components/TextFieldGroup/TextFieldGroup.tokens.ts
  • packages/plasma-new-hope/src/components/TextFieldGroup/variations/_size/base.ts
  • packages/plasma-new-hope/src/components/_Icon/Icons/IconDisclosureDownFill.tsx
  • packages/plasma-new-hope/src/examples/components/Autocomplete/Autocomplete.config.ts
  • packages/plasma-new-hope/src/examples/components/Combobox/Combobox.config.ts
  • packages/plasma-new-hope/src/examples/components/Select/Select.config.ts
  • packages/plasma-new-hope/src/examples/components/TextField/TextField.clear.config.ts
  • packages/plasma-new-hope/src/examples/components/TextField/TextField.config.ts
  • packages/plasma-new-hope/src/examples/components/TextField/TextField.stories.tsx
  • packages/plasma-new-hope/src/examples/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/plasma-tokens-b2b/package.json
  • packages/plasma-tokens-b2c/package.json
  • packages/plasma-tokens-core/package.json
  • packages/plasma-tokens-web/package.json
  • packages/plasma-tokens/data/themes/sdds_insol_next.json
  • packages/plasma-tokens/data/themes/sdds_platform_ai.json
  • packages/plasma-tokens/package.json
  • packages/plasma-typo/package.json
  • packages/plasma-web/package.json
  • packages/plasma-web/src/components/Autocomplete/Autocomplete.config.ts
  • packages/plasma-web/src/components/Calendar/Calendar.config.ts
  • packages/plasma-web/src/components/Calendar/CalendarDouble.config.ts
  • packages/plasma-web/src/components/Combobox/Combobox.config.ts
  • packages/plasma-web/src/components/DatePicker/DatePicker.config.ts
  • packages/plasma-web/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/plasma-web/src/components/Select/Select.config.ts
  • packages/plasma-web/src/components/TextField/TextField.clear.config.ts
  • packages/plasma-web/src/components/TextField/TextField.config.ts
  • packages/plasma-web/src/components/TextField/TextField.stories.tsx
  • packages/plasma-web/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/sdds-bizcom/package.json
  • packages/sdds-bizcom/src/components/Autocomplete/Autocomplete.config.ts
  • packages/sdds-bizcom/src/components/Calendar/Calendar.config.ts
  • packages/sdds-bizcom/src/components/Calendar/CalendarDouble.config.ts
  • packages/sdds-bizcom/src/components/Combobox/Combobox.config.ts
  • packages/sdds-bizcom/src/components/DatePicker/DatePicker.config.ts
  • packages/sdds-bizcom/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/sdds-bizcom/src/components/Select/Select.config.ts
  • packages/sdds-bizcom/src/components/TextField/TextField.clear.config.ts
  • packages/sdds-bizcom/src/components/TextField/TextField.config.ts
  • packages/sdds-bizcom/src/components/TextField/TextField.stories.tsx
  • packages/sdds-bizcom/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/sdds-cs/package.json
  • packages/sdds-cs/src/components/Autocomplete/Autocomplete.config.ts
  • packages/sdds-cs/src/components/Calendar/Calendar.config.ts
  • packages/sdds-cs/src/components/Calendar/CalendarDouble.config.ts
  • packages/sdds-cs/src/components/Combobox/Combobox.config.ts
  • packages/sdds-cs/src/components/DatePicker/DatePicker.config.ts
  • packages/sdds-cs/src/components/Select/Select.config.ts
  • packages/sdds-cs/src/components/TextField/TextField.clear.config.ts
  • packages/sdds-cs/src/components/TextField/TextField.config.ts
  • packages/sdds-cs/src/components/TextField/TextField.stories.tsx
  • packages/sdds-dfa/package.json
  • packages/sdds-dfa/src/components/Autocomplete/Autocomplete.config.ts
  • packages/sdds-dfa/src/components/Calendar/Calendar.config.ts
  • packages/sdds-dfa/src/components/Calendar/CalendarDouble.config.ts
  • packages/sdds-dfa/src/components/Combobox/Combobox.config.ts
  • packages/sdds-dfa/src/components/DatePicker/DatePicker.config.ts
  • packages/sdds-dfa/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/sdds-dfa/src/components/Select/Select.config.ts
  • packages/sdds-dfa/src/components/TextField/TextField.clear.config.ts
  • packages/sdds-dfa/src/components/TextField/TextField.config.ts
  • packages/sdds-dfa/src/components/TextField/TextField.stories.tsx
  • packages/sdds-dfa/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/sdds-finai/package.json
  • packages/sdds-finai/scripts/replaceStyledComponentsToEmotion.js
  • packages/sdds-finai/src/components/Autocomplete/Autocomplete.config.ts
  • packages/sdds-finai/src/components/Autocomplete/Autocomplete.viewMode.config.ts
  • packages/sdds-finai/src/components/Calendar/Calendar.config.ts
  • packages/sdds-finai/src/components/Calendar/CalendarDouble.config.ts
  • packages/sdds-finai/src/components/Combobox/Combobox.config.ts
  • packages/sdds-finai/src/components/Combobox/Combobox.viewMode.config.ts
  • packages/sdds-finai/src/components/DatePicker/DatePicker.config.ts
  • packages/sdds-finai/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/sdds-finai/src/components/Select/Select.config.ts
  • packages/sdds-finai/src/components/Select/Select.viewMode.config.ts
  • packages/sdds-finai/src/components/TextField/TextField.clear.config.ts
  • packages/sdds-finai/src/components/TextField/TextField.config.ts
  • packages/sdds-finai/src/components/TextField/TextField.stories.tsx
  • packages/sdds-finai/src/components/TextField/TextField.viewMode.config.ts
  • packages/sdds-finai/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/sdds-icons/package.json
  • packages/sdds-insol-next/package.json
  • packages/sdds-insol-next/src/components/Autocomplete/Autocomplete.config.ts
  • packages/sdds-insol-next/src/components/Calendar/Calendar.config.ts
  • packages/sdds-insol-next/src/components/Calendar/CalendarDouble.config.ts
  • packages/sdds-insol-next/src/components/Combobox/Combobox.config.ts
  • packages/sdds-insol-next/src/components/DatePicker/DatePicker.config.ts
  • packages/sdds-insol-next/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/sdds-insol-next/src/components/Select/Select.config.ts
  • packages/sdds-insol-next/src/components/TextField/TextField.clear.config.ts
  • packages/sdds-insol-next/src/components/TextField/TextField.config.ts
  • packages/sdds-insol-next/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/sdds-insol/package.json
  • packages/sdds-insol/src/components/Autocomplete/Autocomplete.config.ts
  • packages/sdds-insol/src/components/Calendar/Calendar.config.ts
  • packages/sdds-insol/src/components/Calendar/CalendarDouble.config.ts
  • packages/sdds-insol/src/components/Combobox/Combobox.config.ts
  • packages/sdds-insol/src/components/DatePicker/DatePicker.config.ts
  • packages/sdds-insol/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/sdds-insol/src/components/Select/Select.config.ts
  • packages/sdds-insol/src/components/TextField/TextField.clear.config.ts
  • packages/sdds-insol/src/components/TextField/TextField.config.ts
  • packages/sdds-insol/src/components/TextField/TextField.stories.tsx
  • packages/sdds-insol/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/sdds-netology/package.json
  • packages/sdds-netology/src/components/Autocomplete/Autocomplete.config.ts
  • packages/sdds-netology/src/components/Calendar/Calendar.config.ts
  • packages/sdds-netology/src/components/Calendar/CalendarDouble.config.ts
  • packages/sdds-netology/src/components/Combobox/Combobox.config.ts
  • packages/sdds-netology/src/components/DatePicker/DatePicker.config.ts
  • packages/sdds-netology/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/sdds-netology/src/components/Select/Select.config.ts
  • packages/sdds-netology/src/components/TextField/TextField.clear.config.ts
  • packages/sdds-netology/src/components/TextField/TextField.config.ts
  • packages/sdds-netology/src/components/TextField/TextField.stories.tsx
  • packages/sdds-netology/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/sdds-os/package.json
  • packages/sdds-os/src/components/Autocomplete/Autocomplete.config.ts
  • packages/sdds-os/src/components/Calendar/Calendar.config.ts
  • packages/sdds-os/src/components/Calendar/CalendarDouble.config.ts
  • packages/sdds-os/src/components/Combobox/Combobox.config.ts
  • packages/sdds-os/src/components/DatePicker/DatePicker.config.ts
  • packages/sdds-os/src/components/Select/Select.config.ts
  • packages/sdds-os/src/components/TextField/TextField.config.ts
  • packages/sdds-os/src/components/TextField/TextField.stories.tsx
  • packages/sdds-os/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/sdds-platform-ai/package.json
  • packages/sdds-platform-ai/src/components/Autocomplete/Autocomplete.config.ts
  • packages/sdds-platform-ai/src/components/Calendar/CalendarBase.config.ts
  • packages/sdds-platform-ai/src/components/Calendar/CalendarDouble.config.ts
  • packages/sdds-platform-ai/src/components/Combobox/Combobox.config.ts
  • packages/sdds-platform-ai/src/components/DatePicker/DatePicker.config.ts
  • packages/sdds-platform-ai/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/sdds-platform-ai/src/components/Select/Select.config.ts
  • packages/sdds-platform-ai/src/components/TextField/TextField.config.ts
  • packages/sdds-platform-ai/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/sdds-sbcom/package.json
  • packages/sdds-sbcom/src/components/Autocomplete/Autocomplete.config.ts
  • packages/sdds-sbcom/src/components/Avatar/Avatar.config.tsx
  • packages/sdds-sbcom/src/components/Avatar/Avatar.stories.tsx
  • packages/sdds-sbcom/src/components/Button/Button.config.ts
  • packages/sdds-sbcom/src/components/Button/Button.stories.tsx
  • packages/sdds-sbcom/src/components/Calendar/Calendar.config.ts
  • packages/sdds-sbcom/src/components/Calendar/CalendarDouble.config.ts
  • packages/sdds-sbcom/src/components/Checkbox/Checkbox.config.ts
  • packages/sdds-sbcom/src/components/Chip/Chip.config.tsx
  • packages/sdds-sbcom/src/components/Combobox/Combobox.config.ts
  • packages/sdds-sbcom/src/components/DatePicker/DatePicker.config.ts
  • packages/sdds-sbcom/src/components/DatePicker/DatePicker.stories.tsx
  • packages/sdds-sbcom/src/components/DatePicker/DatePickerRange.stories.tsx
  • packages/sdds-sbcom/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/sdds-sbcom/src/components/Divider/Divider.config.ts
  • packages/sdds-sbcom/src/components/IconButton/IconButton.config.ts
  • packages/sdds-sbcom/src/components/IconButton/IconButton.stories.tsx
  • packages/sdds-sbcom/src/components/Indicator/Indicator.config.ts
  • packages/sdds-sbcom/src/components/Indicator/Indicator.config.tsx
  • packages/sdds-sbcom/src/components/Indicator/Indicator.stories.tsx
  • packages/sdds-sbcom/src/components/Loader/Loader.config.ts
  • packages/sdds-sbcom/src/components/Loader/Loader.stories.tsx
  • packages/sdds-sbcom/src/components/Modal/Modal.config.ts
  • packages/sdds-sbcom/src/components/Modal/Modal.stories.tsx
  • packages/sdds-sbcom/src/components/Radiobox/Radiobox.config.ts
  • packages/sdds-sbcom/src/components/Select/Select.config.ts
  • packages/sdds-sbcom/src/components/Switch/Switch.config.ts
  • packages/sdds-sbcom/src/components/TextArea/TextArea.clear.config.ts
  • packages/sdds-sbcom/src/components/TextArea/TextArea.config.ts
  • packages/sdds-sbcom/src/components/TextArea/TextArea.stories.tsx
  • packages/sdds-sbcom/src/components/TextArea/TextArea.tsx
  • packages/sdds-sbcom/src/components/TextField/TextField.clear.config.ts
  • packages/sdds-sbcom/src/components/TextField/TextField.config.ts
  • packages/sdds-sbcom/src/components/TextField/TextField.search.config.ts
  • packages/sdds-sbcom/src/components/TextField/TextField.stories.tsx
  • packages/sdds-sbcom/src/components/TextField/TextField.tsx
  • packages/sdds-sbcom/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/sdds-sbcom/src/components/Tokens/Typography/Typography.stories.tsx
  • packages/sdds-sbcom/src/components/Tooltip/Tooltip.config.ts
  • packages/sdds-sbcom/src/components/Tooltip/Tooltip.stories.tsx
  • packages/sdds-sbcom/src/components/Typography/Body.config.ts
  • packages/sdds-sbcom/src/components/Typography/Typography.stories.tsx
  • packages/sdds-sbcom/src/components/Typography/Typography.tsx
  • packages/sdds-sbcom/src/components/Typography/index.ts
  • packages/sdds-scan/package.json
  • packages/sdds-scan/src/components/Autocomplete/Autocomplete.config.ts
  • packages/sdds-scan/src/components/Calendar/Calendar.config.ts
  • packages/sdds-scan/src/components/Calendar/CalendarDouble.config.ts
  • packages/sdds-scan/src/components/Combobox/Combobox.config.ts
  • packages/sdds-scan/src/components/DatePicker/DatePicker.config.ts
  • packages/sdds-scan/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/sdds-scan/src/components/Select/Select.config.ts
  • packages/sdds-scan/src/components/TextField/TextField.config.ts
  • packages/sdds-scan/src/components/TextField/TextField.stories.tsx
  • packages/sdds-scan/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/sdds-serv/package.json
  • packages/sdds-serv/src/components/Autocomplete/Autocomplete.config.ts
  • packages/sdds-serv/src/components/Calendar/Calendar.config.ts
  • packages/sdds-serv/src/components/Calendar/CalendarDouble.config.ts
  • packages/sdds-serv/src/components/Combobox/Combobox.config.ts
  • packages/sdds-serv/src/components/DatePicker/DatePicker.config.ts
  • packages/sdds-serv/src/components/DateTimePicker/DateTimePicker.config.ts
  • packages/sdds-serv/src/components/Select/Select.config.ts
  • packages/sdds-serv/src/components/TextField/TextField.clear.config.ts
  • packages/sdds-serv/src/components/TextField/TextField.config.ts
  • packages/sdds-serv/src/components/TextField/TextField.stories.tsx
  • packages/sdds-serv/src/components/TextFieldGroup/TextFieldGroup.config.ts
  • packages/themes/core-themes/package.json
  • packages/themes/plasma-themes/package.json
  • packages/themes/sdds-themes/package.json
  • utils/api-tests/package.json
  • utils/plasma-cy-utils/package.json
  • utils/plasma-docs-ui/package.json
  • utils/plasma-sb-utils/package.json
  • utils/plasma-tokens-utils/package.json
  • website/plasma-b2c-docs/docs/components/Divider.mdx
  • website/plasma-b2c-docs/package.json
  • website/plasma-giga-docs/package.json
  • website/plasma-homeds-docs/package.json
  • website/plasma-web-docs/docs/components/Divider.mdx
  • website/plasma-web-docs/package.json
  • website/plasma-website/package.json
  • website/plasma-website/public/data/plasma-asdk.json
  • website/plasma-website/public/data/plasma-b2c.json
  • website/plasma-website/public/data/plasma-giga.json
  • website/plasma-website/public/data/plasma-homeds.json
  • website/plasma-website/public/data/plasma-new-hope.json
  • website/plasma-website/public/data/plasma-web.json
  • website/plasma-website/public/data/sdds-bizcom.json
  • website/plasma-website/public/data/sdds-cs.json
  • website/plasma-website/public/data/sdds-dfa.json
  • website/plasma-website/public/data/sdds-finai.json
  • website/plasma-website/public/data/sdds-insol-next.json
  • website/plasma-website/public/data/sdds-insol.json
  • website/plasma-website/public/data/sdds-netology.json
  • website/plasma-website/public/data/sdds-os.json
  • website/plasma-website/public/data/sdds-platform-ai.json
  • website/plasma-website/public/data/sdds-sbcom.json
  • website/plasma-website/public/data/sdds-scan.json
  • website/plasma-website/public/data/sdds-serv.json
  • website/sdds-bizcom-docs/package.json
  • website/sdds-cs-docs/docs/components/Divider.mdx
  • website/sdds-cs-docs/package.json
  • website/sdds-dfa-docs/package.json
  • website/sdds-finai-docs/package.json
  • website/sdds-insol-docs/docs/components/Divider.mdx
  • website/sdds-insol-docs/package.json
  • website/sdds-netology-docs/package.json
  • website/sdds-platform-ai-docs/docs/components/Divider.mdx
  • website/sdds-platform-ai-docs/package.json
  • website/sdds-sbcom-docs/docs/components/Divider.mdx
  • website/sdds-sbcom-docs/docs/components/Typography.mdx
  • website/sdds-sbcom-docs/docusaurus.config.js
  • website/sdds-sbcom-docs/package.json
  • website/sdds-scan-docs/package.json
  • website/sdds-serv-docs/docs/components/Divider.mdx
  • website/sdds-serv-docs/package.json

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.


Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out.

❤️ Share

Comment @coderabbitai help to get the list of available commands.

@TitanKuzmich
TitanKuzmich marked this pull request as ready for review August 18, 2026 13:03
@github-actions

Copy link
Copy Markdown
Contributor

Theme Builder app deployed!

https://plasma.sberdevices.ru/pr/plasma-theme-builder-pr-3060/

@github-actions

Copy link
Copy Markdown
Contributor

Documentation preview deployed!

website: https://plasma.sberdevices.ru/pr/pr-3060/
asdk storybook: https://plasma.sberdevices.ru/pr/pr-3060/asdk-storybook/
b2c storybook: https://plasma.sberdevices.ru/pr/pr-3060/b2c-storybook/
giga storybook: https://plasma.sberdevices.ru/pr/pr-3060/giga-storybook/
homeds storybook: https://plasma.sberdevices.ru/pr/pr-3060/homeds-storybook/
new-hope storybook: https://plasma.sberdevices.ru/pr/pr-3060/new-hope-storybook/
web storybook: https://plasma.sberdevices.ru/pr/pr-3060/web-storybook/
sdds-bizcom storybook: https://plasma.sberdevices.ru/pr/pr-3060/sdds-bizcom-storybook/
sdds-cs storybook: https://plasma.sberdevices.ru/pr/pr-3060/sdds-cs-storybook/
sdds-dfa storybook: https://plasma.sberdevices.ru/pr/pr-3060/sdds-dfa-storybook/
sdds-finai storybook: https://plasma.sberdevices.ru/pr/pr-3060/sdds-finai-storybook/
sdds-insol storybook: https://plasma.sberdevices.ru/pr/pr-3060/sdds-insol-storybook/
sdds-insol-next storybook: https://plasma.sberdevices.ru/pr/pr-3060/sdds-insol-next-storybook/
sdds-netology storybook: https://plasma.sberdevices.ru/pr/pr-3060/sdds-netology-storybook/
sdds-os storybook: https://plasma.sberdevices.ru/pr/pr-3060/sdds-os-storybook/
sdds-platform-ai storybook: https://plasma.sberdevices.ru/pr/pr-3060/sdds-platform-ai-storybook/
sdds-sbcom storybook: https://plasma.sberdevices.ru/pr/pr-3060/sdds-sbcom-storybook/
sdds-scan storybook: https://plasma.sberdevices.ru/pr/pr-3060/sdds-scan-storybook/
sdds-serv storybook: https://plasma.sberdevices.ru/pr/pr-3060/sdds-serv-storybook/

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

changelog-skip Exclude pull request changes from changelog

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant